Skip to content

chore(Button): Move Button CSS module feature flag from staff to ga - #5086

Merged
jonrohan merged 5 commits into
mainfrom
css_module_button_ga
Oct 18, 2024
Merged

chore(Button): Move Button CSS module feature flag from staff to ga#5086
jonrohan merged 5 commits into
mainfrom
css_module_button_ga

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 8, 2024

Copy link
Copy Markdown
Member

Changelog

Changed

Moves ButtonBase css modules feature flag from staff to ga

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Oct 8, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 82fea93

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 8, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 8, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5086 October 8, 2024 18:32 Inactive
@github-actions

github-actionsBot commented Oct 8, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.1 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js97.37 KB (-0.12% 🔽)

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Do you think we need to test this canary at dotcom to make sure we don't introduce any regression to GA via the css modules or is it not necessary here?

@jonrohanjonrohan changed the title Move Button CSS module feature flag from staff to gachore(Button): Move Button CSS module feature flag from staff to gaOct 18, 2024
@jonrohan
jonrohan added this pull request to the merge queueOct 18, 2024
@jonrohan

Copy link
Copy Markdown
MemberAuthor

Do you think we need to test this canary at dotcom to make sure we don't introduce any regression to GA via the css modules or is it not necessary here?

Not necessary

Merged via the queue into main with commit 96ce586Oct 18, 2024
@jonrohan
jonrohan deleted the css_module_button_ga branch October 18, 2024 23:41
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@broccolinisoup