Skip to content

feat(AvatarStack): Add support for className prop - #5097

Merged
randall-krauskopf merged 3 commits into
mainfrom
avatar-stack-classname
Oct 10, 2024
Merged

feat(AvatarStack): Add support for className prop#5097
randall-krauskopf merged 3 commits into
mainfrom
avatar-stack-classname

Conversation

@randall-krauskopf

@randall-krauskopfrandall-krauskopf commented Oct 9, 2024

Copy link
Copy Markdown
Contributor

Related To: https://github.com/github/collaboration-workflows-flex/issues/1297

Changelog

New

Added className prop to the AvatarStack component for CSS Module support.

Changed

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Oct 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c9dd0e0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5097 October 9, 2024 19:48 Inactive
@github-actions

github-actionsBot commented Oct 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.9 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js97.19 KB (-0.05% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5097 October 9, 2024 20:24 Inactive
@randall-krauskopfrandall-krauskopf changed the title add support for className prop in the avatar stackAdd support for className prop in the avatar stackOct 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5097 October 9, 2024 20:44 Inactive
@randall-krauskopfrandall-krauskopf changed the title Add support for className prop in the avatar stackfeat(AvatarStack): Add support for className propOct 9, 2024
@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 9, 2024 20:55
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerOctober 9, 2024 20:55

@camertroncamertron left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Seems ok to me 👍

@randall-krauskopf
randall-krauskopf added this pull request to the merge queueOct 10, 2024
@randall-krauskopf
randall-krauskopf removed this pull request from the merge queue due to a manual request Oct 10, 2024
@randall-krauskopf
randall-krauskopf added this pull request to the merge queueOct 10, 2024
Merged via the queue into main with commit 2a3c473Oct 10, 2024
@randall-krauskopf
randall-krauskopf deleted the avatar-stack-classname branch October 10, 2024 13:59
@primerprimerBot mentioned this pull request Oct 10, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@randall-krauskopf@camertron