Skip to content

feat: add ButtonBase to experimental - #5099

Merged
joshblack merged 6 commits into
mainfrom
featt/add-button-base-to-experimental
Oct 10, 2024
Merged

feat: add ButtonBase to experimental#5099
joshblack merged 6 commits into
mainfrom
featt/add-button-base-to-experimental

Conversation

@joshblack

Copy link
Copy Markdown
Member

To help with wildcard import migrations, I'm adding ButtonBase to @primer/react/experimental since this is currently being used downstream. This seemed like an appropriate place for now since we don't necessarily want to expose this as a stable component.

Changelog

New

  • Add ButtonBase to @primer/react/experimental

Changed

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerOctober 9, 2024 20:21
@changeset-bot

changeset-botBot commented Oct 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0747989

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 9, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 9, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5099 October 9, 2024 20:24 Inactive
@github-actions

github-actionsBot commented Oct 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.91 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js97.27 KB (+0.05% 🔺)

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This seemed like an appropriate place for now since we don't necessarily want to expose this as a stable component.

Makes sense!

@joshblack
joshblack added this pull request to the merge queueOct 10, 2024
Merged via the queue into main with commit cb21a38Oct 10, 2024
@joshblack
joshblack deleted the featt/add-button-base-to-experimental branch October 10, 2024 15:06
@primerprimerBot mentioned this pull request Oct 10, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@broccolinisoup