Skip to content

fix(ProgressBar): add aria-valuetext attribute - #5100

Merged
francinelucca merged 6 commits into
mainfrom
3458-prcprogressbar-best-practice-add-a-human-understandable-value-announcement
Oct 16, 2024
Merged

fix(ProgressBar): add aria-valuetext attribute#5100
francinelucca merged 6 commits into
mainfrom
3458-prcprogressbar-best-practice-add-a-human-understandable-value-announcement

Conversation

@francinelucca

@francineluccafrancinelucca commented Oct 9, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3458

Adds aria-valuetext attribute to ProgressBar component to improve accessibility for screen readers by ensuring percentage value (60% vs just 60) gets read.

Changelog

New

  • Adds aria-valuetext attribute to ProgressBar component

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Verify percentage gets read on screen reader

Merge checklist

@francineluccafrancinelucca added component: ProgressBar staff Author is a staff member labels Oct 9, 2024
@changeset-bot

changeset-botBot commented Oct 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f2cfda7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 9, 2024
@github-actions

github-actionsBot commented Oct 9, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.6 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js97.83 KB (-0.11% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5100 October 9, 2024 21:35 Inactive
@francineluccafrancinelucca added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 9, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 9, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/346662

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@francinelucca
francinelucca marked this pull request as ready for review October 10, 2024 23:01
@francinelucca
francinelucca requested a review from a team as a code ownerOctober 10, 2024 23:01
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5100 October 16, 2024 17:50 Inactive
@francinelucca
francinelucca added this pull request to the merge queueOct 16, 2024
Merged via the queue into main with commit b0d858aOct 16, 2024
@francinelucca
francinelucca deleted the 3458-prcprogressbar-best-practice-add-a-human-understandable-value-announcement branch October 16, 2024 18:02
This was referenced Oct 16, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

component: ProgressBarintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff memberstatus: review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@joshblack