Skip to content

Add className prop support to <Textarea> - #5130

Merged
joshblack merged 5 commits into
mainfrom
textarea-classname
Oct 16, 2024
Merged

Add className prop support to <Textarea>#5130
joshblack merged 5 commits into
mainfrom
textarea-classname

Conversation

@randall-krauskopf

Copy link
Copy Markdown
Contributor

Related to: https://github.com/github/collaboration-workflows-flex/issues/1145

Changelog

New

Added className prop to the Textarea component for CSS Module support.

Changed

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Oct 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dd7e5f6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 15, 2024
@github-actions

github-actionsBot commented Oct 15, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.57 KB (-0.03% 🔽)
packages/react/dist/browser.umd.js97.93 KB (+0.1% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5130 October 15, 2024 20:47 Inactive
@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 15, 2024 21:07
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerOctober 15, 2024 21:07
Comment threadpackages/react/src/Textarea/Textarea.docs.json Outdated
Comment thread.changeset/wicked-ties-walk.md Outdated
randall-krauskopfand others added 2 commits October 15, 2024 16:34
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@github.com>
@joshblack
joshblack added this pull request to the merge queueOct 16, 2024
Merged via the queue into main with commit 661e94cOct 16, 2024
@joshblack
joshblack deleted the textarea-classname branch October 16, 2024 17:16
This was referenced Oct 16, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@randall-krauskopf@joshblack