Skip to content

Bug fix: ActionList Group className prop - #5132

Merged
joshblack merged 8 commits into
mainfrom
action-list-heading-bug
Oct 16, 2024
Merged

Bug fix: ActionList Group className prop#5132
joshblack merged 8 commits into
mainfrom
action-list-heading-bug

Conversation

@langermank

@langermanklangermank commented Oct 16, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4241

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Oct 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 37d1326

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5132 October 16, 2024 15:48 Inactive
@github-actions

github-actionsBot commented Oct 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.62 KB (-0.06% 🔽)
packages/react/dist/browser.umd.js97.9 KB (-0.04% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/347345

@primer-integration

primer-integrationBot commented Oct 16, 2024

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5132 October 16, 2024 17:51 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 16, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5132 October 16, 2024 19:00 Inactive
@langermank
langermank marked this pull request as ready for review October 16, 2024 19:08
@langermank
langermank requested review from a team as code ownersOctober 16, 2024 19:08
@langermank
langermank added this pull request to the merge queueOct 16, 2024
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Oct 16, 2024
@joshblack

Copy link
Copy Markdown
Member

@broccolinisoup do you think it'd be possible for this to be added to the current release candidate? Or would this be better in the next one?

@broccolinisoup

Copy link
Copy Markdown
Member

@joshblack yeah go for it 🔥

@joshblack
joshblack added this pull request to the merge queueOct 16, 2024
Merged via the queue into main with commit 623a411Oct 16, 2024
@joshblack
joshblack deleted the action-list-heading-bug branch October 16, 2024 20:48
This was referenced Oct 16, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@joshblack@broccolinisoup