Skip to content

chore(CounterLabel): Move CounterLabel css modules feature flag to staff - #5147

Merged
jonrohan merged 2 commits into
mainfrom
move_counterlabel_to_staff
Oct 22, 2024
Merged

chore(CounterLabel): Move CounterLabel css modules feature flag to staff#5147
jonrohan merged 2 commits into
mainfrom
move_counterlabel_to_staff

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Changelog

Changed

Move CounterLabel css modules feature flag to staff

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Oct 17, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 97c7586

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 17, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 17, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5147 October 17, 2024 23:40 Inactive
@github-actions

github-actionsBot commented Oct 17, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.04 KB (-0.06% 🔽)
packages/react/dist/browser.umd.js97.42 KB (-0.07% 🔽)

@jonrohan
jonrohan marked this pull request as ready for review October 17, 2024 23:42
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 17, 2024 23:42
@jonrohan
jonrohan enabled auto-merge October 22, 2024 20:18
@jonrohan
jonrohan added this pull request to the merge queueOct 22, 2024
Merged via the queue into main with commit e421f32Oct 22, 2024
@jonrohan
jonrohan deleted the move_counterlabel_to_staff branch October 22, 2024 20:28
@primerprimerBot mentioned this pull request Oct 22, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack