Skip to content

feat(Details): Convert Details to css module behind feature flag - #5167

Merged
jonrohan merged 6 commits into
mainfrom
css_modules_details
Oct 24, 2024
Merged

feat(Details): Convert Details to css module behind feature flag#5167
jonrohan merged 6 commits into
mainfrom
css_modules_details

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4136

Changelog

Changed

Convert Details to css module behind feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 21, 2024
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 21, 2024 21:29
@changeset-bot

changeset-botBot commented Oct 21, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1b1eeb8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 21, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 21, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5167 October 21, 2024 21:32 Inactive
@github-actions

github-actionsBot commented Oct 21, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.12 KB (+0.09% 🔺)
packages/react/dist/browser.umd.js97.5 KB (+0.12% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/348108

@primer
primerBot requested a review from a team as a code ownerOctober 21, 2024 21:37
@primer
primerBot requested a review from langermankOctober 21, 2024 21:37
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Oct 21, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5167 October 21, 2024 21:40 Inactive
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good! Left one question but not a blocker.

colorScheme: theme,
},
})
const stories: Array<{title: string; id: string}> = [

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Is that a new practise we're exploring that includes the stories in the e2e test file? 👀

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Yeah @joshblack started doing this in some of the other tests and thought I would unify around it

Merged via the queue into main with commit 07b75e7Oct 24, 2024
@jonrohan
jonrohan deleted the css_modules_details branch October 24, 2024 01:02
@primerprimerBot mentioned this pull request Oct 24, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@broccolinisoup