Skip to content

Allow primitives dep to be if version 9 OR 10 and add it to peerDep - #5177

Merged
joshblack merged 1 commit into
mainfrom
update-primitves-dep
Nov 4, 2024
Merged

Allow primitives dep to be if version 9 OR 10 and add it to peerDep#5177
joshblack merged 1 commit into
mainfrom
update-primitves-dep

Conversation

@lukasoppermann

@lukasoppermannlukasoppermann commented Oct 27, 2024

Copy link
Copy Markdown
Contributor

Allow primitives dep to be if version 9 OR 10 and add it to peerDep

Closeshttps://github.com/github/primer/issues/4322

Changelog

Changed

dependency @primer/primitives to allow both major version 9 and 10 and move it from dependency to peer dendency

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@lukasoppermann
lukasoppermann requested a review from a team as a code ownerOctober 27, 2024 11:00
@changeset-bot

changeset-botBot commented Oct 27, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3b67144

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 27, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5177 October 27, 2024 11:03 Inactive
@github-actions

github-actionsBot commented Oct 27, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.48 KB (0%)
packages/react/dist/browser.umd.js98.91 KB (0%)

Comment threadpackages/react/package.json Outdated
@lukasoppermannlukasoppermann removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 4, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5177 November 4, 2024 20:29 Inactive
@joshblack
joshblack added this pull request to the merge queueNov 4, 2024
Merged via the queue into main with commit 26b54d8Nov 4, 2024
@joshblack
joshblack deleted the update-primitves-dep branch November 4, 2024 20:41
@primerprimerBot mentioned this pull request Nov 4, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@lukasoppermann@keithamus@joshblack@langermank