Skip to content

feat(Radio): Convert Radio to css modules behind feature flag - #5187

Merged
jonrohan merged 4 commits into
mainfrom
css_modules_convert_radio
Oct 30, 2024
Merged

feat(Radio): Convert Radio to css modules behind feature flag#5187
jonrohan merged 4 commits into
mainfrom
css_modules_convert_radio

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 29, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4298

Changelog

Changed

Convert Radio to css modules behind feature flag

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Randall Krauskopf <randall-krauskopf@users.noreply.github.com>
Co-authored-by: Hussam Ghazzi <hussam-i-am@github.com>
@changeset-bot

changeset-botBot commented Oct 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3051589

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 29, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 29, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5187 October 29, 2024 16:31 Inactive
@github-actions

github-actionsBot commented Oct 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.42 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js97.79 KB (+0.03% 🔺)

@primer-integration

primer-integrationBot commented Oct 29, 2024

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@jonrohanjonrohan changed the title Convert radio to CSS modulesfeat(Radio): Convert Radio to css modules behind feature flagOct 29, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349165

@jonrohan
jonrohan marked this pull request as ready for review October 29, 2024 20:40
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 29, 2024 20:40

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM to me if things are 💯 downstream, the only thing that I can think of is if we need to wrap nested styles in :where() because their specificity may win out if anyone is trying to target checked or disabled styles

@jonrohan
jonrohan enabled auto-merge October 30, 2024 19:08
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5187 October 30, 2024 19:10 Inactive
@jonrohan
jonrohan added this pull request to the merge queueOct 30, 2024
Merged via the queue into main with commit 2ab7b9eOct 30, 2024
@jonrohan
jonrohan deleted the css_modules_convert_radio branch October 30, 2024 19:24
@primerprimerBot mentioned this pull request Oct 30, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack