Skip to content

feat(Spinner): Convert Spinner to CSS module behind feature flag - #5188

Merged
randall-krauskopf merged 5 commits into
mainfrom
spinner-css-modules
Oct 30, 2024
Merged

feat(Spinner): Convert Spinner to CSS module behind feature flag#5188
randall-krauskopf merged 5 commits into
mainfrom
spinner-css-modules

Conversation

@randall-krauskopf

@randall-krauskopfrandall-krauskopf commented Oct 29, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4293

Changelog

Changed

Convert spinner component to css module behind feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Co-authored-by: Hussam Ghazzi <hussam-i-am@github.com>
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
@changeset-bot

changeset-botBot commented Oct 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 46f9dcb

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 29, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5188 October 29, 2024 17:15 Inactive
@github-actions

github-actionsBot commented Oct 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.34 KB (+0.11% 🔺)
packages/react/dist/browser.umd.js97.63 KB (+0.13% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5188 October 29, 2024 18:20 Inactive
@randall-krauskopfrandall-krauskopf changed the title Convert spinner component to CSS modulesfeat(Spinner): Convert Spinner to CSS module behind feature flagOct 29, 2024
@primer-integration

primer-integrationBot commented Oct 29, 2024

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349176

@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 30, 2024 14:26
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerOctober 30, 2024 14:26
@randall-krauskopf
randall-krauskopf marked this pull request as draft October 30, 2024 14:26
@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 30, 2024 14:44
@randall-krauskopf
randall-krauskopf added this pull request to the merge queueOct 30, 2024
Merged via the queue into main with commit 573ae51Oct 30, 2024
@randall-krauskopf
randall-krauskopf deleted the spinner-css-modules branch October 30, 2024 16:52
@primerprimerBot mentioned this pull request Oct 30, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@randall-krauskopf@jonrohan