Skip to content

Improve the click area for horizontal FormControls (checkboxes and radios) - #5190

Merged
mperrotti merged 8 commits into
mainfrom
mp/improve-checkbox-radio-click-area
Nov 23, 2024
Merged

Improve the click area for horizontal FormControls (checkboxes and radios)#5190
mperrotti merged 8 commits into
mainfrom
mp/improve-checkbox-radio-click-area

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/3539

Changelog

New

Changed

  • Changes <label> wrapper of FormControl to put padding directly on the label so the entire row is clickable

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@mperrotti
mperrotti requested a review from a team as a code ownerOctober 29, 2024 18:32
@changeset-bot

changeset-botBot commented Oct 29, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1574716

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 29, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5190 October 29, 2024 18:34 Inactive
@github-actions

github-actionsBot commented Oct 29, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.54 KB (+0.05% 🔺)
packages/react/dist/browser.umd.js102.83 KB (-0.02% 🔽)

@mperrottimperrotti added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5190 November 22, 2024 23:26 Inactive
@primer
primerBot requested a review from a team as a code ownerNovember 22, 2024 23:35
@primer
primerBot requested a review from langermankNovember 22, 2024 23:35
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5190 November 22, 2024 23:38 Inactive
@mperrottimperrotti added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5190 November 22, 2024 23:45 Inactive
@mperrottimperrotti added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Nov 22, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 23, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5190 November 23, 2024 00:05 Inactive
@mperrotti
mperrotti added this pull request to the merge queueNov 23, 2024
Merged via the queue into main with commit 32c4fe1Nov 23, 2024
@mperrotti
mperrotti deleted the mp/improve-checkbox-radio-click-area branch November 23, 2024 00:26
@primerprimerBot mentioned this pull request Nov 23, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@mperrotti@TylerJDev