Skip to content

feat(SkeletonBox): Convert SkeletonBox to CSS modules behind the feature flag - #5195

Merged
jonrohan merged 7 commits into
mainfrom
css_modules_skeleton_box
Nov 1, 2024
Merged

feat(SkeletonBox): Convert SkeletonBox to CSS modules behind the feature flag#5195
jonrohan merged 7 commits into
mainfrom
css_modules_skeleton_box

Conversation

@jonrohan

@jonrohanjonrohan commented Oct 30, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4129
Closeshttps://github.com/github/primer/issues/4360
Closeshttps://github.com/github/primer/issues/4361

Changelog

New

Changed

  • feat(SkeletonAvatar): Convert SkeletonAvatar to CSS modules behind the feature flag
  • feat(SkeletonBox): Convert SkeletonBox to CSS modules behind the feature flag
  • feat(SkeletonText): Convert SkeletonText to CSS modules behind the feature flag

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Oct 30, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6ad126a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Oct 30, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 30, 2024
@github-actions

github-actionsBot commented Oct 30, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.2 KB (+0.13% 🔺)
packages/react/dist/browser.umd.js98.52 KB (+0.19% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5195 October 31, 2024 17:41 Inactive
@jonrohan
jonrohan marked this pull request as ready for review October 31, 2024 17:56
@jonrohan
jonrohan requested a review from a team as a code ownerOctober 31, 2024 17:56
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349502

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@jonrohan
jonrohan enabled auto-merge October 31, 2024 18:22
@jonrohan
jonrohan disabled auto-merge October 31, 2024 18:22
@jonrohan
jonrohan added this pull request to the merge queueNov 1, 2024
Merged via the queue into main with commit d349cfcNov 1, 2024
@jonrohan
jonrohan deleted the css_modules_skeleton_box branch November 1, 2024 15:26
@primerprimerBot mentioned this pull request Nov 1, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack