Skip to content

feat(Pagehead): Convert Pagehead component to CSS Modules behind feature flag - #5197

Merged
randall-krauskopf merged 4 commits into
mainfrom
pagehead-css-module
Oct 31, 2024
Merged

feat(Pagehead): Convert Pagehead component to CSS Modules behind feature flag#5197
randall-krauskopf merged 4 commits into
mainfrom
pagehead-css-module

Conversation

@randall-krauskopf

@randall-krauskopfrandall-krauskopf commented Oct 31, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4137

Changelog

Changed

Migrated Pagehead component to CSS Modules behind feature flag.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

@changeset-bot

changeset-botBot commented Oct 31, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 58c2e07

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 31, 2024
@github-actions

github-actionsBot commented Oct 31, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js97.39 KB (0%)
packages/react/dist/browser.umd.js97.76 KB (0%)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349432

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 31, 2024 16:37
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerOctober 31, 2024 16:37
}

export function shouldNotAcceptSystemProps() {
// @ts-expect-error system props should not be accepted

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Torn on this change, cause it now will technically accept the system props we were blocking before. I think the change comes from the toggleStyledComponent somewhere. But this is temporary. When we remove the feature flag it will be an error again.

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I can hold off on this and try and find a solution if you like? I don't mind, the temporary change could be in place for awhile depending on how things go?

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Na, if we work fast enough on this, it won't make a difference. and I don't think folks are adding these attributes in dotcom much anymore. if at all

@randall-krauskopf
randall-krauskopf added this pull request to the merge queueOct 31, 2024
Merged via the queue into main with commit ad84d4fOct 31, 2024
@randall-krauskopf
randall-krauskopf deleted the pagehead-css-module branch October 31, 2024 18:29
@primerprimerBot mentioned this pull request Oct 31, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@randall-krauskopf@jonrohan