Skip to content

feat(SubNav): Convert SubNav to CSS modules behind team feature flag - #5202

Merged
hussam-i-am merged 9 commits into
mainfrom
hussam-i-am/css-modules-subnav
Nov 4, 2024
Merged

feat(SubNav): Convert SubNav to CSS modules behind team feature flag#5202
hussam-i-am merged 9 commits into
mainfrom
hussam-i-am/css-modules-subnav

Conversation

@hussam-i-am

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4363

Changelog

New

Changed

Convert SubNav to CSS modules behind team feature flag

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@hussam-i-amhussam-i-am self-assigned this Oct 31, 2024
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerOctober 31, 2024 17:50
@changeset-bot

changeset-botBot commented Oct 31, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1ec2c06

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 31, 2024
}

.SubNav-octicon {
color: ${get('colors.fg.muted')};

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Removed the SubNav-octicon stylings since it does not appear that we add this class anywhere, let me know if that is correct.

@github-actions

github-actionsBot commented Oct 31, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.55 KB (+0.16% 🔺)
packages/react/dist/browser.umd.js98.82 KB (+0.07% 🔺)

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I think this component might be missing VRT tests all together (I couldn't find any). You might want to open a separate PR to get that merged in first to check against your PR https://github.com/primer/react/blob/main/contributor-docs/testing.md#components

And then for the dev stories, you would want to add those to VRT as well 😄

Comment on lines +65 to +66
/* stylelint-disable-next-line primer/colors */
border-color: var(--bgColor-accent-emphasis);

@langermanklangermankOct 31, 2024

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
/* stylelint-disable-next-line primer/colors */
border-color:var(--bgColor-accent-emphasis);
border-color:var(--borderColor-accent-emphasis);

Should actually be the same end result

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@langermank Reverted this change since VRT was failing due to the border having a different color with Dark Hight Contrast

Screenshot 2024-11-03 at 11 12 01 PM

transition: background-color 0.2s ease;
}

&:is([data-selected]) {

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Should this one be :where?

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

When I tried using :where, the hover/focus styling was overriding the selected styling, which is why I switched to :is. Open to alternative solutions.

@primer-integration

primer-integrationBot commented Oct 31, 2024

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5202 November 4, 2024 04:30 Inactive
@hussam-i-amhussam-i-am added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@primer
primerBot requested a review from a team as a code ownerNovember 4, 2024 04:48
@primer
primerBot requested a review from emilybrickNovember 4, 2024 04:48
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5202 November 4, 2024 04:50 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349556

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Seems good to go ⚡

@hussam-i-am
hussam-i-am added this pull request to the merge queueNov 4, 2024
Merged via the queue into main with commit 293c574Nov 4, 2024
@hussam-i-am
hussam-i-am deleted the hussam-i-am/css-modules-subnav branch November 4, 2024 18:33
@primerprimerBot mentioned this pull request Nov 4, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@hussam-i-am@jonrohan@langermank