Skip to content

feat(InlineMessage): Convert InlineMessage component to CSS Modules behind feature flag - #5204

Merged
randall-krauskopf merged 12 commits into
mainfrom
inline-message-css-module
Nov 5, 2024
Merged

feat(InlineMessage): Convert InlineMessage component to CSS Modules behind feature flag#5204
randall-krauskopf merged 12 commits into
mainfrom
inline-message-css-module

Conversation

@randall-krauskopf

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4130

Changelog

Changed

Migrated InlineMessage component to CSS Modules behind feature flag.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

@changeset-bot

changeset-botBot commented Oct 31, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bd32155

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Oct 31, 2024
@github-actions

github-actionsBot commented Oct 31, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.5 KB (0%)
packages/react/dist/browser.umd.js98.91 KB (0%)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349537

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@randall-krauskopf
randall-krauskopf marked this pull request as ready for review October 31, 2024 20:09
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerOctober 31, 2024 20:09
Comment threadpackages/react/src/InlineMessage/InlineMessage.tsx Outdated
Comment threadpackages/react/src/InlineMessage/InlineMessage.tsx Outdated
Comment threadpackages/react/src/InlineMessage/InlineMessage.dev.stories.tsx Outdated
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5204 November 4, 2024 15:55 Inactive
@randall-krauskopfrandall-krauskopf added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@randall-krauskopfrandall-krauskopf removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@randall-krauskopfrandall-krauskopf added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@primer
primerBot requested a review from a team as a code ownerNovember 4, 2024 17:04
@primer
primerBot requested a review from tbenningNovember 4, 2024 17:04
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5204 November 4, 2024 17:07 Inactive
@randall-krauskopfrandall-krauskopf added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5204 November 4, 2024 20:03 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@randall-krauskopf
randall-krauskopf added this pull request to the merge queueNov 5, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Nov 5, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5204 November 5, 2024 22:43 Inactive
@randall-krauskopf
randall-krauskopf added this pull request to the merge queueNov 5, 2024
Merged via the queue into main with commit 209c9b0Nov 5, 2024
@randall-krauskopf
randall-krauskopf deleted the inline-message-css-module branch November 5, 2024 23:20
@primerprimerBot mentioned this pull request Nov 5, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@randall-krauskopf@hussam-i-am@joshblack@langermank