Skip to content

feat(Stack): Convert Stack to CSS modules behind feature flag - #5214

Merged
jonrohan merged 3 commits into
mainfrom
css_modules_stack
Nov 1, 2024
Merged

feat(Stack): Convert Stack to CSS modules behind feature flag#5214
jonrohan merged 3 commits into
mainfrom
css_modules_stack

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4362

Changelog

Changed

Convert Stack to CSS modules behind feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerNovember 1, 2024 20:15
@changeset-bot

changeset-botBot commented Nov 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ff73594

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 1, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 1, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5214 November 1, 2024 20:18 Inactive
@github-actions

github-actionsBot commented Nov 1, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.28 KB (+0.09% 🔺)
packages/react/dist/browser.umd.js98.58 KB (-0.1% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/349768

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@jonrohan
jonrohan added this pull request to the merge queueNov 1, 2024
Merged via the queue into main with commit 916ed95Nov 1, 2024
@jonrohan
jonrohan deleted the css_modules_stack branch November 1, 2024 21:07
@primerprimerBot mentioned this pull request Nov 1, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack