Skip to content

fix(Blankslate): update styles to center description text - #5229

Merged
joshblack merged 5 commits into
mainfrom
fix/update-blankslate-to-center-text
Nov 5, 2024
Merged

fix(Blankslate): update styles to center description text#5229
joshblack merged 5 commits into
mainfrom
fix/update-blankslate-to-center-text

Conversation

@joshblack

Copy link
Copy Markdown
Member

The text in Blankslate should be text-align: center instead of justified left 🔍

Changelog

New

Changed

  • Set text-align: center on the Blankslate container

Removed

Rollout strategy

  • Patch release

@joshblack
joshblack requested a review from a team as a code ownerNovember 4, 2024 22:56
@changeset-bot

changeset-botBot commented Nov 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e325c08

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 4, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 4, 2024

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:chef-kiss:

@joshblack

Copy link
Copy Markdown
MemberAuthor

@jonrohan I noticed that in PVC this style is set at the container level and I wanted to ask if we should do it similarly or, since this component accepts children, if we should scope it to the description/heading?

The only concern I would have is that anything rendered as children to Blankslate would also have text alignment this way (which could be intended but I wanted to double check).

@github-actions

github-actionsBot commented Nov 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.48 KB (0%)
packages/react/dist/browser.umd.js98.91 KB (0%)

@jonrohan

Copy link
Copy Markdown
Member

since this component accepts children, if we should scope it to the description/heading?

Good point, I would probably scope to just the .Description class. Since scoping on the container will require children to have text-align: left to reset the style

@primer
primerBot requested a review from a team as a code ownerNovember 4, 2024 23:04
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5229 November 4, 2024 23:07 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5229 November 5, 2024 16:51 Inactive
@joshblack
joshblack added this pull request to the merge queueNov 5, 2024
Merged via the queue into main with commit 33c5086Nov 5, 2024
@joshblack
joshblack deleted the fix/update-blankslate-to-center-text branch November 5, 2024 17:05
@primerprimerBot mentioned this pull request Nov 5, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@jonrohan