Skip to content

feat(CheckBoxOrRadioGroup): Convert CheckBoxOrRadioGroup to CSS modules behind feature flag - #5275

Merged
jonrohan merged 6 commits into
mainfrom
css_modules_checkbox_or_radio_group
Nov 14, 2024
Merged

feat(CheckBoxOrRadioGroup): Convert CheckBoxOrRadioGroup to CSS modules behind feature flag#5275
jonrohan merged 6 commits into
mainfrom
css_modules_checkbox_or_radio_group

Conversation

@jonrohan

@jonrohanjonrohan commented Nov 13, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4305

Changelog

New

Changed

Convert CheckBoxOrRadioGroup to CSS modules behind feature flag

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

jonrohanand others added 2 commits November 13, 2024 20:46
Co-authored-by: Marie Lucca <francinelucca@users.noreply.github.com>
Co-authored-by: Tyler Jones <TylerJDev@users.noreply.github.com>
@changeset-bot

changeset-botBot commented Nov 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 09c3b3f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 13, 2024
@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5275 November 13, 2024 21:12 Inactive
@github-actions

github-actionsBot commented Nov 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.94 KB (+0.22% 🔺)
packages/react/dist/browser.umd.js99.31 KB (+0.24% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5275 November 13, 2024 22:05 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 13, 2024
@jonrohan
jonrohan marked this pull request as ready for review November 13, 2024 22:38
@jonrohan
jonrohan requested review from a team as code ownersNovember 13, 2024 22:38
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/351114

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@jonrohan
jonrohan marked this pull request as draft November 14, 2024 20:10
@jonrohanjonrohan changed the title chore(CheckBoxOrRadioGroup): Convert CheckBoxOrRadioGroup to CSS modules behind feature flagfeat(CheckBoxOrRadioGroup): Convert CheckBoxOrRadioGroup to CSS modules behind feature flagNov 14, 2024
@jonrohan
jonrohan marked this pull request as ready for review November 14, 2024 21:40
@jonrohan
jonrohan added this pull request to the merge queueNov 14, 2024
Merged via the queue into main with commit d1b7bceNov 14, 2024
@jonrohan
jonrohan deleted the css_modules_checkbox_or_radio_group branch November 14, 2024 23:28
@primerprimerBot mentioned this pull request Nov 14, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@hussam-i-am