Skip to content

chore(Checkbox): Move Checkbox CSS module feature flag from staff to ga - #5280

Merged
jonrohan merged 2 commits into
mainfrom
css_modules_checkbox_ga
Nov 13, 2024
Merged

chore(Checkbox): Move Checkbox CSS module feature flag from staff to ga #5280
jonrohan merged 2 commits into
mainfrom
css_modules_checkbox_ga

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Changelog

New

Changed

chore(Checkbox): Move Checkbox CSS module feature flag from staff to ga

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerNovember 13, 2024 22:52
@changeset-bot

changeset-botBot commented Nov 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 81989dd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 13, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Nov 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js98.75 KB (-0.04% 🔽)
packages/react/dist/browser.umd.js98.99 KB (0%)

@jonrohan
jonrohan added this pull request to the merge queueNov 13, 2024
Merged via the queue into main with commit 973130eNov 13, 2024
@jonrohan
jonrohan deleted the css_modules_checkbox_ga branch November 13, 2024 23:15
@primerprimerBot mentioned this pull request Nov 13, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank