Uh oh!
There was an error while loading. Please reload this page.
feat(AvatarStack): Convert AvatarStack to CSS modules behind team feature flag - #5299
Conversation
Co-authored-by: Josh Black <joshblack@github.com>
🦋 Changeset detectedLatest commit: 7235818 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks! |
jonrohan
commented
Nov 20, 2024
Ok, this is ready again. This time I've fixed the issues we saw when shipping last time https://github.com/github/github/pull/351862 |
francinelucca
left a comment
There was a problem hiding this comment.
Mostly LGTM, just a few comment cleanup suggestions and one question:
I see integration PR is failing, is that just because it needs the new changes on the separate dotcom PR? 👀
non-blocking: I'm wondering if this story is working as supposed to. It behaves the same in prod but are the sizes even doing anything here? 😅
Uh oh!
There was an error while loading. Please reload this page.
Uh oh!
There was an error while loading. Please reload this page.
Uh oh!
There was an error while loading. Please reload this page.
| line-height: 1; | ||
| border-radius: 50%; | ||
| /* stylelint-disable-next-line primer/box-shadow */ | ||
| box-shadow: 0 0 0 1px var(--avatar-borderColor); |
There was a problem hiding this comment.
I forget the explanation, The box-shadow "border" wasn't showing up in snapshots before, but when avatar stack was converted it started to fail.
Uh oh!
There was an error while loading. Please reload this page.
Co-authored-by: Marie Lucca <40550942+francinelucca@users.noreply.github.com> Co-authored-by: Hussam Ghazzi <hussam-i-am@github.com>
👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/351802 |
🔴 golden-jobs completed with status |
Closeshttps://github.com/github/primer/issues/4024
Changelog
Changed
Update
AvatarStackcomponent to use CSS modules behind the feature flag primer_react_css_modules_teamRollout strategy
Testing & Reviewing
Merge checklist