Skip to content

feat(Popover): Convert Popover to CSS modules behind team feature flag - #5300

Merged
francinelucca merged 9 commits into
mainfrom
francinelucca/4139-update-popover-to-css-modules
Nov 18, 2024
Merged

feat(Popover): Convert Popover to CSS modules behind team feature flag#5300
francinelucca merged 9 commits into
mainfrom
francinelucca/4139-update-popover-to-css-modules

Conversation

@francinelucca

@francineluccafrancinelucca commented Nov 15, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4139

Changelog

New

  • Popover dev story to test sx and style props with VRT
  • Popover module css
  • Snapshots for new dev story

Changed

  • Refactor Popover e2e tests and add new dev story
  • Refactor Popover to use CSS modules when team feature flag is turned on

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@francineluccafrancinelucca added status: wip update snapshots 🤖 Command that updates VRT snapshots on the pull request component: Popover labels Nov 15, 2024
@changeset-bot

changeset-botBot commented Nov 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3cdcda9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 15, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 15, 2024
@github-actions

github-actionsBot commented Nov 15, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js100.25 KB (+0.22% 🔺)
packages/react/dist/browser.umd.js100.53 KB (+0.18% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5300 November 15, 2024 03:34 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 15, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5300 November 15, 2024 03:42 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/351364

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 15, 2024
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Nov 15, 2024
@francinelucca
francinelucca marked this pull request as ready for review November 15, 2024 18:51
@francinelucca
francinelucca requested review from a team as code ownersNovember 15, 2024 18:51

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Awesome ✨

@francinelucca
francinelucca added this pull request to the merge queueNov 18, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Nov 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5300 November 18, 2024 15:55 Inactive
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5300 November 18, 2024 18:08 Inactive
@francinelucca
francinelucca added this pull request to the merge queueNov 18, 2024
Merged via the queue into main with commit 65802fcNov 18, 2024
@francinelucca
francinelucca deleted the francinelucca/4139-update-popover-to-css-modules branch November 18, 2024 19:59
@primerprimerBot mentioned this pull request Nov 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

component: Popoverintegration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff memberstatus: review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@francinelucca@jonrohan@langermank