Skip to content

feat(ProgressBar): Convert ProgressBar to CSS modules behind team feature flag - #5304

Merged
TylerJDev merged 12 commits into
mainfrom
tylerjdev/progressbar-css-modules
Nov 19, 2024
Merged

feat(ProgressBar): Convert ProgressBar to CSS modules behind team feature flag#5304
TylerJDev merged 12 commits into
mainfrom
tylerjdev/progressbar-css-modules

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Nov 15, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4304

Changelog

New

  • Converts ProgressBar to CSS modules behind team feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Nov 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a7161a5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 15, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Nov 15, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js100.62 KB (+0.23% 🔺)
packages/react/dist/browser.umd.js100.93 KB (+0.12% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/351565

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5304 November 18, 2024 18:08 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review November 18, 2024 18:24
@TylerJDev
TylerJDev requested a review from a team as a code ownerNovember 18, 2024 18:24
border-radius: var(--borderRadius-small);
gap: 2px;

&[data-progress-display='inline'] {

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Something we're doing for the data- attribute selectors is wrapping in :where() which brings the specificity down on the selector so it's easier to override in dotcom.

Suggested change
&[data-progress-display='inline'] {
&:where([data-progress-display='inline']) {

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great 👍🏻

@TylerJDev
TylerJDev added this pull request to the merge queueNov 18, 2024
@TylerJDev
TylerJDev removed this pull request from the merge queue due to a manual request Nov 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5304 November 18, 2024 22:56 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 19, 2024
@primer
primerBot requested a review from a team as a code ownerNovember 19, 2024 14:29
@primer
primerBot requested a review from emilybrickNovember 19, 2024 14:29
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 19, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5304 November 19, 2024 14:32 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueNov 19, 2024
Merged via the queue into main with commit a3fc488Nov 19, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/progressbar-css-modules branch November 19, 2024 14:57
@primerprimerBot mentioned this pull request Nov 19, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@jonrohan