Skip to content

feat(Overlay): Convert Overlay to CSS modules behind team feature flag - #5310

Merged
francinelucca merged 27 commits into
mainfrom
francinelucca/4135-update-overlay-to-css-modules
Dec 11, 2024
Merged

feat(Overlay): Convert Overlay to CSS modules behind team feature flag#5310
francinelucca merged 27 commits into
mainfrom
francinelucca/4135-update-overlay-to-css-modules

Conversation

@francinelucca

@francineluccafrancinelucca commented Nov 18, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4135

Changelog

New

  • Overlay module css

Changed

  • Refactor Overlay to use CSS modules when team feature flag is turned on
  • Update SelectPanel import to new Overlay base component name (StyledOverlay -> BaseOverlay)

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Nov 18, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0a54dfc

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Nov 18, 2024
@francineluccafrancinelucca added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed staff Author is a staff member labels Nov 18, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Nov 19, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.49 KB (+0.05% 🔺)
packages/react/dist/browser.umd.js102.86 KB (+0.1% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5310 November 19, 2024 19:00 Inactive
@langermanklangermank added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Nov 19, 2024
Comment threadpackages/react/src/Overlay/Overlay.module.css Outdated
Comment threadpackages/react/src/Overlay/Overlay.module.css Outdated
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
@francineluccafrancinelucca removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 19, 2024
@francineluccafrancinelucca added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 19, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5310 November 27, 2024 18:57 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5310 November 27, 2024 22:05 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 27, 2024
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/354358

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Dec 10, 2024
@francinelucca
francinelucca marked this pull request as ready for review December 10, 2024 19:48
@francinelucca
francinelucca requested a review from a team as a code ownerDecember 10, 2024 19:48
@francinelucca
francinelucca added this pull request to the merge queueDec 11, 2024
@primerprimerBot mentioned this pull request Dec 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@francinelucca@joshblack@langermank