Skip to content

feat(SelectPanel2): Convert SelectPanel2 to CSS modules - #5325

Merged
hussam-i-am merged 9 commits into
mainfrom
hussam-i-am/select-panel-2-css-modules
Nov 25, 2024
Merged

feat(SelectPanel2): Convert SelectPanel2 to CSS modules#5325
hussam-i-am merged 9 commits into
mainfrom
hussam-i-am/select-panel-2-css-modules

Conversation

@hussam-i-am

Copy link
Copy Markdown
Contributor

Changelog

New

Changed

  • Converts SelectPanel to CSS modules without feature flag since it is only used internally and does not currently accept an sx or className prop

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@hussam-i-amhussam-i-am self-assigned this Nov 21, 2024
@changeset-bot

changeset-botBot commented Nov 21, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 09de060

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 21, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Nov 21, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.49 KB (0%)
packages/react/dist/browser.umd.js102.85 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5325 November 22, 2024 00:27 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Nov 22, 2024
@hussam-i-am
hussam-i-am marked this pull request as ready for review November 22, 2024 16:08
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerNovember 22, 2024 16:08
Comment threadpackages/react/src/experimental/SelectPanel2/SelectPanel.tsx Outdated

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Mostly LGTM, just one question around classNames.
I'm wondering if we want to rethink shipping under a FF, I do see this being pulled in from /experimental in dotcom unless I'm reading this wrong?

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 22, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Nov 22, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@hussam-i-am

Copy link
Copy Markdown
ContributorAuthor

@francinelucca Readded the sx props and update classNames to pass through from props

@github-actionsgithub-actionsBot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 22, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5325 November 22, 2024 19:30 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/352234

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks good! one non-blocking suggestion

<Box sx={{display: 'flex', alignItems: 'center', gap: 2}}>
<Checkbox id={checkboxId} sx={{marginTop: 0}} {...props} />
<InputLabel htmlFor={checkboxId} sx={{fontSize: 0}}>
<Box

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

nit: should we render theses Boxes as native divs when enabled?

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I believe it defaults to a div and doesn't need it in those cases

@hussam-i-am
hussam-i-am added this pull request to the merge queueNov 25, 2024
Merged via the queue into main with commit cdc4da1Nov 25, 2024
@hussam-i-am
hussam-i-am deleted the hussam-i-am/select-panel-2-css-modules branch November 25, 2024 16:04
@primerprimerBot mentioned this pull request Nov 25, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@hussam-i-am@francinelucca