Skip to content

feat(ConfirmationDialog): Convert ConfirmationDialog to CSS modules behind team feature flag - #5331

Merged
TylerJDev merged 11 commits into
mainfrom
tylerjdev/css-modules-confirmation-dialog
Dec 10, 2024
Merged

feat(ConfirmationDialog): Convert ConfirmationDialog to CSS modules behind team feature flag #5331
TylerJDev merged 11 commits into
mainfrom
tylerjdev/css-modules-confirmation-dialog

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Nov 25, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4134

Changelog

New

  • Converts ConfirmationDialog to CSS modules behind team feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Nov 25, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3afa403

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Nov 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 November 25, 2024 15:44 Inactive
@github-actions

github-actionsBot commented Nov 25, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.56 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js103.9 KB (+0.1% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 November 25, 2024 15:52 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 25, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 November 25, 2024 16:08 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 November 25, 2024 18:48 Inactive
@TylerJDevTylerJDev added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Nov 25, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Nov 25, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 November 25, 2024 20:13 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/352653

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 3, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@TylerJDevTylerJDev removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5331 December 3, 2024 00:07 Inactive
@TylerJDevTylerJDev added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Dec 3, 2024
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 3, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

const ConfirmationBody: React.FC<React.PropsWithChildren<DialogProps>> = ({children}) => {
return <StyledConfirmationBody>{children}</StyledConfirmationBody>
const enabled = useFeatureFlag(CSS_MODULES_FEATURE_FLAG)
return <StyledConfirmationBody className={enabled && classes.ConfirmationBody}>{children}</StyledConfirmationBody>

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Since I don't believe we allow custom styling for this component, I didn't bother adding a className prop. Not sure if we should always add one regardless or not if the component doesn't utilize sx or custom styles?

@TylerJDev
TylerJDev marked this pull request as ready for review December 3, 2024 15:04
@TylerJDev
TylerJDev requested review from a team as code ownersDecember 3, 2024 15:04
@TylerJDev
TylerJDev added this pull request to the merge queueDec 10, 2024
@TylerJDev
TylerJDev added this pull request to the merge queueDec 10, 2024
Merged via the queue into main with commit e39dddbDec 10, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/css-modules-confirmation-dialog branch December 10, 2024 17:41
@primerprimerBot mentioned this pull request Dec 10, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@jonrohan