Skip to content

refactor(FormControl): update existing code for sub-components to prep for migration - #5342

Merged
joshblack merged 10 commits into
mainfrom
refactor/form-control-css-1
Dec 4, 2024
Merged

refactor(FormControl): update existing code for sub-components to prep for migration#5342
joshblack merged 10 commits into
mainfrom
refactor/form-control-css-1

Conversation

@joshblack

@joshblackjoshblack commented Dec 2, 2024

Copy link
Copy Markdown
Member

Update FormControl sub-components to be easier to migrate to CSS modules.

Changelog

New

Changed

  • Update the following FormControl sub-components to a different styled component style:
    • Label
    • Caption
    • LeadingVisual
    • Validation

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Dec 2, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: dbd9e32

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 2, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.91 KB (+0.18% 🔺)
packages/react/dist/browser.umd.js104.33 KB (+0.27% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/353326

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 3, 2024
@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@joshblack
joshblack marked this pull request as ready for review December 3, 2024 17:56
@joshblack
joshblack requested a review from a team as a code ownerDecember 3, 2024 17:56
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5342 December 3, 2024 17:58 Inactive
Comment threadpackage.json
Comment threadpackages/react/src/internal/components/InputLabel.tsx Outdated
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
Merged via the queue into main with commit 39df71eDec 4, 2024
@joshblack
joshblack deleted the refactor/form-control-css-1 branch December 4, 2024 16:42
@primerprimerBot mentioned this pull request Dec 4, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@jonrohan