Skip to content

docs: fix *.docs.json stories - #5345

Merged
francinelucca merged 4 commits into
mainfrom
francinelucca/4433-bug-fix-storybook-examples-in-primer-docs-for-prc-actionbar-and-datatable
Dec 3, 2024
Merged

docs: fix *.docs.json stories#5345
francinelucca merged 4 commits into
mainfrom
francinelucca/4433-bug-fix-storybook-examples-in-primer-docs-for-prc-actionbar-and-datatable

Conversation

@francinelucca

@francineluccafrancinelucca commented Dec 3, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4433

Some story IDs are listed incorrectly in their respective *.docs.json, this PR fixes them so that they can be pulled into primer.style storybook embed

Changelog

Changed

  • Update story ids in some components' *.docs.json files

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Test deployment: https://primer-e313264adc-26441320.drafts.github.io/components/data-table/react/alpha

Merge checklist

@changeset-bot

changeset-botBot commented Dec 3, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4e6e6b7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Dec 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5345 December 3, 2024 02:54 Inactive
@github-actions

github-actionsBot commented Dec 3, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.73 KB (0%)
packages/react/dist/browser.umd.js104.06 KB (0%)

@francineluccafrancinelucca added the docs Documentation label Dec 3, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5345 December 3, 2024 21:43 Inactive
@francinelucca
francinelucca marked this pull request as ready for review December 3, 2024 21:53
@francinelucca
francinelucca requested a review from a team as a code ownerDecember 3, 2024 21:53
…s-in-primer-docs-for-prc-actionbar-and-datatable

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice ✨

@francinelucca
francinelucca added this pull request to the merge queueDec 3, 2024
Merged via the queue into main with commit e231b5fDec 3, 2024
@francinelucca
francinelucca deleted the francinelucca/4433-bug-fix-storybook-examples-in-primer-docs-for-prc-actionbar-and-datatable branch December 3, 2024 22:48
@primerprimerBot mentioned this pull request Dec 3, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
francinelucca added a commit that referenced this pull request Dec 5, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

docsDocumentationstaffAuthor is a staff memberstatus: review neededstatus: wip

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@TylerJDev