Skip to content

feat(BaseStyles): Convert BaseStyles to CSS modules behind team feature flag - #5356

Closed
JelloBagel wants to merge 7 commits into
mainfrom
jellobagel-basestyles-css-modules
Closed

feat(BaseStyles): Convert BaseStyles to CSS modules behind team feature flag#5356
JelloBagel wants to merge 7 commits into
mainfrom
jellobagel-basestyles-css-modules

Conversation

@JelloBagel

@JelloBagelJelloBagel commented Dec 4, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4358

Changelog

Changed

  • Convert BaseStyles to CSS modules behind primer_react_css_modules_team feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 4, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0c6494f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 4, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 4, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.71 KB (-0.06% 🔽)
packages/react/dist/browser.umd.js104.23 KB (+0.13% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 18:56 Inactive
@JelloBagelJelloBagel added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 4, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 20:36 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 20:58 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 21:22 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 21:52 Inactive
@JelloBagelJelloBagel added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 4, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 4, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 22:32 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 22:52 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 23:00 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 23:14 Inactive
@JelloBagel
JelloBagelforce-pushed the jellobagel-basestyles-css-modules branch from 7a133ed to 584547dCompareDecember 4, 2024 23:23
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5356 December 4, 2024 23:26 Inactive
Comment threade2e/components/SelectPanel.test.ts
@JelloBagel
JelloBagel marked this pull request as ready for review December 4, 2024 23:58
@JelloBagel
JelloBagel requested a review from a team as a code ownerDecember 4, 2024 23:58
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 5, 2024
@JelloBagel
JelloBagel marked this pull request as draft December 5, 2024 00:41
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/353683

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@JelloBagel

JelloBagel commented Dec 5, 2024

Copy link
Copy Markdown
ContributorAuthor

Remaking to clean up PR: #5361

@JelloBagel
JelloBagel deleted the jellobagel-basestyles-css-modules branch December 5, 2024 19:10
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@JelloBagel