Skip to content

feat(ActionList.Heading): Convert to CSS Modules - #5367

Merged
langermank merged 9 commits into
mainfrom
al-css-heading
Dec 6, 2024
Merged

feat(ActionList.Heading): Convert to CSS Modules#5367
langermank merged 9 commits into
mainfrom
al-css-heading

Conversation

@langermank

@langermanklangermank commented Dec 5, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4449

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 5, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 37b507d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 5, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5367 December 5, 2024 22:51 Inactive
@github-actions

github-actionsBot commented Dec 5, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.76 KB (+0.1% 🔺)
packages/react/dist/browser.umd.js104.11 KB (+0.1% 🔺)

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 5, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 5, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5367 December 5, 2024 23:34 Inactive
@langermank
langermank marked this pull request as ready for review December 5, 2024 23:46
@langermank
langermank requested review from a team as code ownersDecember 5, 2024 23:46
Comment threade2e/components/ActionList.test.ts Outdated
@jonrohanjonrohan added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5367 December 6, 2024 01:11 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5367 December 6, 2024 01:21 Inactive
Comment threadpackages/react/src/ActionList/Heading.test.tsx Outdated
@langermank
langermank added this pull request to the merge queueDec 6, 2024
Merged via the queue into main with commit 844e41fDec 6, 2024
@langermank
langermank deleted the al-css-heading branch December 6, 2024 19:37
@primerprimerBot mentioned this pull request Dec 6, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan