Skip to content

chore(Pagehead): Move Pagehead component css module feature flag to ga - #5378

Merged
jonrohan merged 3 commits into
mainfrom
css_module_pagehead_ga
Dec 11, 2024
Merged

chore(Pagehead): Move Pagehead component css module feature flag to ga#5378
jonrohan merged 3 commits into
mainfrom
css_module_pagehead_ga

Conversation

@jonrohan

@jonrohanjonrohan commented Dec 6, 2024

Copy link
Copy Markdown
Member

Changelog

New

Changed

Move Pagehead component css module feature flag to ga

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

This won't require any integration tests since it's a feature flag change.

Merge checklist

@changeset-bot

changeset-botBot commented Dec 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ba7e36e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 6, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.39 KB (0%)
packages/react/dist/browser.umd.js102.66 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5378 December 6, 2024 18:28 Inactive
@jonrohanjonrohan removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@jonrohan
jonrohan marked this pull request as ready for review December 6, 2024 18:30
@jonrohan
jonrohan requested a review from a team as a code ownerDecember 6, 2024 18:30
@jonrohan
jonrohan enabled auto-merge December 6, 2024 18:35
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5378 December 6, 2024 23:16 Inactive
@jonrohanjonrohan removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@jonrohan
jonrohan added this pull request to the merge queueDec 11, 2024
Merged via the queue into main with commit 5a8138aDec 11, 2024
@jonrohan
jonrohan deleted the css_module_pagehead_ga branch December 11, 2024 15:57
@primerprimerBot mentioned this pull request Dec 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@joshblack