Skip to content

chore(Stack): Move Stack component css module feature flag to ga - #5384

Merged
jonrohan merged 2 commits into
mainfrom
css_module_stack_ga
Dec 16, 2024
Merged

chore(Stack): Move Stack component css module feature flag to ga#5384
jonrohan merged 2 commits into
mainfrom
css_module_stack_ga

Conversation

@jonrohan

@jonrohanjonrohan commented Dec 6, 2024

Copy link
Copy Markdown
Member

Changelog

New

Changed

Move Stack component css module feature flag to ga

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

This won't require any integration tests since it's a feature flag change.

Merge checklist

@changeset-bot

changeset-botBot commented Dec 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7af154e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 6, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.66 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js104.01 KB (-0.01% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5384 December 6, 2024 18:29 Inactive
@jonrohanjonrohan removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@jonrohan
jonrohan marked this pull request as ready for review December 6, 2024 18:34
@jonrohan
jonrohan requested a review from a team as a code ownerDecember 6, 2024 18:34
@jonrohan
jonrohan enabled auto-merge December 6, 2024 18:35
@jonrohan
jonrohan added this pull request to the merge queueDec 16, 2024
Merged via the queue into main with commit eedbdbdDec 16, 2024
@jonrohan
jonrohan deleted the css_module_stack_ga branch December 16, 2024 16:46
@primerprimerBot mentioned this pull request Dec 16, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@hussam-i-am