Skip to content

chore(DialogV1): Move DialogV1 component css module feature flag to staff - #5395

Merged
jonrohan merged 2 commits into
mainfrom
css_module_dialogv1_staff
Dec 6, 2024
Merged

chore(DialogV1): Move DialogV1 component css module feature flag to staff#5395
jonrohan merged 2 commits into
mainfrom
css_module_dialogv1_staff

Conversation

@jonrohan

@jonrohanjonrohan commented Dec 6, 2024

Copy link
Copy Markdown
Member

Changelog

New

Changed

Move DialogV1 component css module feature flag to staff

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

This won't require any integration tests since it's a feature flag change.

Merge checklist

@changeset-bot

changeset-botBot commented Dec 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ad26c3d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Dec 6, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5395 December 6, 2024 18:57 Inactive
@github-actions

github-actionsBot commented Dec 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.61 KB (0%)
packages/react/dist/browser.umd.js104.11 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5395 December 6, 2024 19:01 Inactive
@jonrohanjonrohan removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 6, 2024
@jonrohan
jonrohan marked this pull request as ready for review December 6, 2024 19:03
@jonrohan
jonrohan requested a review from a team as a code ownerDecember 6, 2024 19:03
@jonrohan
jonrohan enabled auto-merge December 6, 2024 19:03

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:shipit:

@jonrohan
jonrohan added this pull request to the merge queueDec 6, 2024
Merged via the queue into main with commit a90da75Dec 6, 2024
@jonrohan
jonrohan deleted the css_module_dialogv1_staff branch December 6, 2024 20:08
@primerprimerBot mentioned this pull request Dec 6, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@francinelucca