Skip to content

Remove min-width on leading visuals + add to Selection in ActionList - #5411

Merged
langermank merged 13 commits into
mainfrom
rm-min-width-al
Dec 11, 2024
Merged

Remove min-width on leading visuals + add to Selection in ActionList#5411
langermank merged 13 commits into
mainfrom
rm-min-width-al

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Just another fine tuning detail to prep for CSS Modules. This width is causing a bit of blowout sometimes breaking the row layout with 8px gaps.

@changeset-bot

changeset-botBot commented Dec 10, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ccd3096

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 10, 2024
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 10, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 10, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.52 KB (-0.03% 🔽)
packages/react/dist/browser.umd.js102.88 KB (+0.03% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 10, 2024 23:37 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 10, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 10, 2024 23:39 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 10, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 10, 2024 23:48 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 10, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 10, 2024 23:58 Inactive
@langermank
langermank marked this pull request as ready for review December 10, 2024 23:58
@langermank
langermank requested review from a team as code ownersDecember 10, 2024 23:58
@langermanklangermank changed the title Remove min-width on leading visuals in ActionListRemove min-width on leading visuals + add to Selection in ActionListDec 11, 2024
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 11, 2024 00:16 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5411 December 11, 2024 00:26 Inactive
@langermank
langermank added this pull request to the merge queueDec 11, 2024
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Dec 11, 2024
@langermank
langermank added this pull request to the merge queueDec 11, 2024
Merged via the queue into main with commit 74d3de4Dec 11, 2024
@langermank
langermank deleted the rm-min-width-al branch December 11, 2024 00:56
@primerprimerBot mentioned this pull request Dec 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@mperrotti