Skip to content

feat(SideNav): Convert SideNav to CSS modules behind feature flag - #5413

Merged
JelloBagel merged 7 commits into
mainfrom
jellobagel-sidenav-cssmodules
Dec 11, 2024
Merged

feat(SideNav): Convert SideNav to CSS modules behind feature flag#5413
JelloBagel merged 7 commits into
mainfrom
jellobagel-sidenav-cssmodules

Conversation

@JelloBagel

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4132

Changelog

Changed

  • Convert SideNav to CSS modules behind primer_react_css_modules_team feature flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 11, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d17a5d8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5413 December 11, 2024 00:32 Inactive
@github-actions

github-actionsBot commented Dec 11, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.74 KB (+0.23% 🔺)
packages/react/dist/browser.umd.js102.98 KB (0%)

@JelloBagelJelloBagel added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5413 December 11, 2024 00:50 Inactive
@JelloBagel
JelloBagelforce-pushed the jellobagel-sidenav-cssmodules branch from 97e2bb7 to 193fea4CompareDecember 11, 2024 15:58
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5413 December 11, 2024 16:00 Inactive
@JelloBagelJelloBagel added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5413 December 11, 2024 16:36 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/354611

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 11, 2024
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@JelloBagel
JelloBagel marked this pull request as ready for review December 11, 2024 17:53
@JelloBagel
JelloBagel requested a review from a team as a code ownerDecember 11, 2024 17:53
@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Dec 11, 2024
Base automatically changed from jellobagel-sidenav-vrt to mainDecember 11, 2024 18:49
@JelloBagel
JelloBagel requested a review from a team as a code ownerDecember 11, 2024 18:49
@JelloBagel
JelloBagel added this pull request to the merge queueDec 11, 2024
Merged via the queue into main with commit 9f5a9acDec 11, 2024
@JelloBagel
JelloBagel deleted the jellobagel-sidenav-cssmodules branch December 11, 2024 19:05
@primerprimerBot mentioned this pull request Dec 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@JelloBagel@joshblack