Skip to content

fix(PageHeader): Correctly add display prop - #5415

Merged
randall-krauskopf merged 5 commits into
mainfrom
page-header-title-bug
Dec 12, 2024
Merged

fix(PageHeader): Correctly add display prop#5415
randall-krauskopf merged 5 commits into
mainfrom
page-header-title-bug

Conversation

@randall-krauskopf

Copy link
Copy Markdown
Contributor

Closes: https://github.com/github/primer/issues/4481

Changelog

Changed

  • Fixed bug for incorrectly displaying display prop for title

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@changeset-bot

changeset-botBot commented Dec 11, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 686eae8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 11, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@randall-krauskopfrandall-krauskopf changed the title correctly add display propfix(PageHeader): Correctly add display propDec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5415 December 11, 2024 20:01 Inactive
@github-actions

github-actionsBot commented Dec 11, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.47 KB (-0.01% 🔽)
packages/react/dist/browser.umd.js103.87 KB (+0.18% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/354656

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 11, 2024
@randall-krauskopf
randall-krauskopf marked this pull request as ready for review December 11, 2024 20:26
@randall-krauskopf
randall-krauskopf requested a review from a team as a code ownerDecember 11, 2024 20:26
@randall-krauskopf

Copy link
Copy Markdown
ContributorAuthor

Integration tests are failing due to unrelated errors

Comment threadpackages/react/src/PageHeader/PageHeader.module.css Outdated
Co-authored-by: Marie Lucca <40550942+francinelucca@users.noreply.github.com>
@randall-krauskopf
randall-krauskopf deleted the page-header-title-bug branch December 12, 2024 15:16
@primerprimerBot mentioned this pull request Dec 12, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@randall-krauskopf@francinelucca