Skip to content

ActionList color bug fixes - #5416

Merged
langermank merged 9 commits into
mainfrom
small-bug-fixes-al
Dec 11, 2024
Merged

ActionList color bug fixes#5416
langermank merged 9 commits into
mainfrom
small-bug-fixes-al

Conversation

@langermank

@langermanklangermank commented Dec 11, 2024

Copy link
Copy Markdown
Contributor

Few more visual bug fixes pre-ActionList CSS

beforeafterwhat changed
imageimagefocus color fix
imageimagefocus color fix
imageimagecheckbox color fix
imageimagecheckbox border color

@changeset-bot

changeset-botBot commented Dec 11, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 70c1857

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @langermank, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

🤖 Beep boop! This comment was added automatically by github/accessibility-alt-text-bot.

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 11, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @langermank, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

🤖 Beep boop! This comment was added automatically by github/accessibility-alt-text-bot.

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions

github-actionsBot commented Dec 11, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.35 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js103.74 KB (+0.09% 🔺)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5416 December 11, 2024 20:19 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 11, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5416 December 11, 2024 20:29 Inactive
@langermank
langermank marked this pull request as ready for review December 11, 2024 20:37
@langermank
langermank requested review from a team as code ownersDecember 11, 2024 20:37
@langermank
langermank added this pull request to the merge queueDec 11, 2024
Merged via the queue into main with commit 1dcc534Dec 11, 2024
@langermank
langermank deleted the small-bug-fixes-al branch December 11, 2024 20:49
@primerprimerBot mentioned this pull request Dec 11, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@mperrotti