Skip to content

bugfix(Stack): Fix typo in custom media variable used in Stack component - #5419

Merged
jonrohan merged 2 commits into
mainfrom
fix_stack_viewport_typo
Dec 12, 2024
Merged

bugfix(Stack): Fix typo in custom media variable used in Stack component#5419
jonrohan merged 2 commits into
mainfrom
fix_stack_viewport_typo

Conversation

@jonrohan

Copy link
Copy Markdown
Member

This fixes an issue brought up https://github.slack.com/archives/C01L618AEP9/p1733943072984009

The variable I was using for the custom media was a typo and wasn't compiling into the actual value.

Changelog

New

Changed

Fix typo in custom media variable used in Stack component

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerDecember 12, 2024 03:22
@changeset-bot

changeset-botBot commented Dec 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fd1cdc8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Dec 12, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 12, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5419 December 12, 2024 03:24 Inactive
@github-actions

github-actionsBot commented Dec 12, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.45 KB (0%)
packages/react/dist/browser.umd.js103.88 KB (0%)

@jonrohan
jonrohan added this pull request to the merge queueDec 12, 2024
Merged via the queue into main with commit 73da985Dec 12, 2024
@jonrohan
jonrohan deleted the fix_stack_viewport_typo branch December 12, 2024 14:36
@primerprimerBot mentioned this pull request Dec 12, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@TylerJDev