Skip to content

Add className back to TokenBase - #5420

Merged
TylerDixon merged 3 commits into
mainfrom
tylerdixon-add-back-classname
Dec 12, 2024
Merged

Add className back to TokenBase#5420
TylerDixon merged 3 commits into
mainfrom
tylerdixon-add-back-classname

Conversation

@TylerDixon

Copy link
Copy Markdown
Contributor

Prior to #5271, the className in TokenBase was part of the ...rest variables that were spread onto the token base. Now that the className is directly referenced as part of the feature-flagged token, we need to directly reference it for the non-feature flagged token.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Create a Token with a class name, that class should be on the resulting rendered token.

Merge checklist

@changeset-bot

changeset-botBot commented Dec 12, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d02eca8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 12, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 12, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.47 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js103.69 KB (-0.19% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5420 December 12, 2024 13:42 Inactive
@TylerDixon
TylerDixon marked this pull request as ready for review December 12, 2024 13:49
@TylerDixon
TylerDixon requested a review from a team as a code ownerDecember 12, 2024 13:49

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Thanks 🙏🏼

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5420 December 12, 2024 14:17 Inactive
@TylerDixon
TylerDixon added this pull request to the merge queueDec 12, 2024
Merged via the queue into main with commit 7a2d486Dec 12, 2024
@TylerDixon
TylerDixon deleted the tylerdixon-add-back-classname branch December 12, 2024 14:35
@primerprimerBot mentioned this pull request Dec 11, 2024
francinelucca added a commit that referenced this pull request Dec 12, 2024
* Add className back to TokenBase
* Adding tests and fixing implementation
* Create dirty-wasps-behave.md
---------
Co-authored-by: Marie Lucca <40550942+francinelucca@users.noreply.github.com>
This was referenced Dec 12, 2024
francinelucca added a commit to francinelucca/react that referenced this pull request Dec 30, 2024
* Add className back to TokenBase
* Adding tests and fixing implementation
* Create dirty-wasps-behave.md
---------
Co-authored-by: Marie Lucca <40550942+francinelucca@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerDixon@francinelucca