Skip to content

chore(NavList) add e2e tests - #5432

Merged
langermank merged 4 commits into
mainfrom
navlist-e2e
Dec 13, 2024
Merged

chore(NavList) add e2e tests#5432
langermank merged 4 commits into
mainfrom
navlist-e2e

Conversation

@langermank

@langermanklangermank commented Dec 13, 2024

Copy link
Copy Markdown
Contributor
  • Add e2e snapshots for NavList
  • Fix markup bug causes a11y failures

Adding snapshots caused an accessibility failure due to some incompatible markup. I added a new prop to try and address that. There are cases where this heading should be a div (I think most cases?) but within NavList it should sometimes be an li.

beforeafter
imageimage

@langermank
langermank requested a review from a team as a code ownerDecember 13, 2024 00:19
@changeset-bot

changeset-botBot commented Dec 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7f1e7af

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermanklangermank added skip changeset This change does not need a changelog update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Dec 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5432 December 13, 2024 00:21 Inactive
@github-actions

github-actionsBot commented Dec 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.23 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js103.57 KB (+0.02% 🔺)

@primer
primerBot requested a review from a team as a code ownerDecember 13, 2024 00:30
@primer
primerBot requested a review from maximedegreveDecember 13, 2024 00:30
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5432 December 13, 2024 00:32 Inactive
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@langermank
langermank requested review from mperrotti and removed request for maximedegreveDecember 13, 2024 15:31
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5432 December 13, 2024 15:33 Inactive
@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @langermank, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

🤖 Beep boop! This comment was added automatically by github/accessibility-alt-text-bot.

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/355064

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024

@TylerJDevTylerJDev left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM! ✨

@langermanklangermank removed the skip changeset This change does not need a changelog label Dec 13, 2024
@langermank
langermank added this pull request to the merge queueDec 13, 2024
Merged via the queue into main with commit 60a2f47Dec 13, 2024
@langermank
langermank deleted the navlist-e2e branch December 13, 2024 16:24
@primerprimerBot mentioned this pull request Dec 13, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@TylerJDev