Skip to content

refactor(Table): update to CSS Modules - #5435

Merged
jonrohan merged 16 commits into
mainfrom
refactor/update-table-to-css-modules
Dec 19, 2024
Merged

refactor(Table): update to CSS Modules#5435
jonrohan merged 16 commits into
mainfrom
refactor/update-table-to-css-modules

Conversation

@joshblack

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4307

Update Table to CSS Modules behind the team feature flag.

Changelog

New

Changed

  • Update Table to CSS Modules behind a feature flag
  • Update Table sub-components to an easier-to-migrate form of styled-components

Removed

Rollout strategy

  • Minor release

@changeset-bot

changeset-botBot commented Dec 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b500884

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5435 December 13, 2024 16:15 Inactive
@github-actions

github-actionsBot commented Dec 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.65 KB (0%)
packages/react/dist/browser.umd.js104.99 KB (0%)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/355809

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 18, 2024
@JelloBagel
JelloBagel marked this pull request as ready for review December 18, 2024 20:34
@JelloBagel
JelloBagel requested a review from a team as a code ownerDecember 18, 2024 20:34
@jonrohan
jonrohan added this pull request to the merge queueDec 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5435 December 18, 2024 23:56 Inactive
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 19, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5435 December 19, 2024 01:30 Inactive
@JelloBagelJelloBagel added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 19, 2024
@JelloBagelJelloBagel self-assigned this Dec 19, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 19, 2024
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 19, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5435 December 19, 2024 19:01 Inactive
@jonrohan
jonrohan added this pull request to the merge queueDec 19, 2024
Merged via the queue into main with commit 5cba52aDec 19, 2024
@jonrohan
jonrohan deleted the refactor/update-table-to-css-modules branch December 19, 2024 19:24
@primerprimerBot mentioned this pull request Dec 19, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants

@joshblack@hussam-i-am@langermank@jonrohan@JelloBagel