Skip to content

test: add check for story ids in generated docs - #5441

Merged
francinelucca merged 16 commits into
mainfrom
test/add-check-for-story-ids
Jan 3, 2025
Merged

test: add check for story ids in generated docs#5441
francinelucca merged 16 commits into
mainfrom
test/add-check-for-story-ids

Conversation

@joshblack

@joshblackjoshblack commented Dec 13, 2024

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/3632

With these changes, CI will fail if a *.docs.json story is not a valid storybook story ID, preventing broken stories on production docs.
Sample error:
error log on broken story id for hidden component

Changelog

New

  • CI step to build storybook and test again generated story IDs in *.docs.json
  • Added logic to check for valid story IDs in generated docs to components-json/build script

Changed

  • Updated storybook build script
  • Updated broken IconButton story ID (the action is working!)
  • Updated broken Hidden story ID (the action is working!)
  • Updated broken InlineMessage story ID (the action is working!)
  • Updated broken Portal story ID (the action is working!)
  • Updated broken TooltipV2 story ID (the action is working!)

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 13, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 134b3d1

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@joshblackjoshblack added skip changeset This change does not need a changelog and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Dec 13, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 December 13, 2024 20:17 Inactive
@github-actions

github-actionsBot commented Dec 13, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.7 KB (0%)
packages/react/dist/browser.umd.js104.15 KB (0%)

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 27, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

Uh oh! @joshblack, the image you shared is missing helpful alt text. Check your pull request body.

Alt text is an invisible description that helps screen readers describe images to blind or low-vision users. If you are using markdown to display images, add your alt text inside the brackets of the markdown image.

Learn more about alt text at Basic writing and formatting syntax: images on GitHub Docs.

🤖 Beep boop! This comment was added automatically by github/accessibility-alt-text-bot.

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 December 27, 2024 21:19 Inactive
@francinelucca
francinelucca marked this pull request as ready for review December 27, 2024 21:30
@francinelucca
francinelucca requested a review from a team as a code ownerDecember 27, 2024 21:30
@francinelucca
francinelucca added this pull request to the merge queueJan 2, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Jan 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 January 2, 2025 18:36 Inactive
@francineluccafrancinelucca added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 2, 2025
@primer
primerBot requested a review from a team as a code ownerJanuary 2, 2025 19:17
@primer
primerBot requested a review from emilybrickJanuary 2, 2025 19:17
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Jan 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 January 2, 2025 19:19 Inactive
@TylerJDevTylerJDev mentioned this pull request Jan 2, 2025
13 tasks
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 January 3, 2025 18:04 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 January 3, 2025 18:10 Inactive
@francinelucca
francineluccaforce-pushed the test/add-check-for-story-ids branch from 7879d45 to 949f52fCompareJanuary 3, 2025 18:11
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5441 January 3, 2025 18:15 Inactive
@francinelucca
francinelucca added this pull request to the merge queueJan 3, 2025
Merged via the queue into main with commit 30e8c31Jan 3, 2025
@francinelucca
francinelucca deleted the test/add-check-for-story-ids branch January 3, 2025 18:42
@primerprimerBot mentioned this pull request Jan 3, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmskip changesetThis change does not need a changelogstaffAuthor is a staff memberstatus: review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@TylerJDev@francinelucca