Skip to content

feat(Autocomplete): Convert Box usage to CSS modules behind feature flag - #5451

Merged
hussam-i-am merged 23 commits into
mainfrom
auto-complete-css-modules
Dec 19, 2024
Merged

feat(Autocomplete): Convert Box usage to CSS modules behind feature flag #5451
hussam-i-am merged 23 commits into
mainfrom
auto-complete-css-modules

Conversation

@randall-krauskopf

@randall-krauskopfrandall-krauskopf commented Dec 16, 2024

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4303

Changelog

New

Changed

  • Updated Box usage to use CSS modules behind a feature flag in Autocomplete

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 69042e4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5451 December 16, 2024 23:17 Inactive
@github-actions

github-actionsBot commented Dec 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js103.35 KB (+0.02% 🔺)
packages/react/dist/browser.umd.js103.71 KB (+0.04% 🔺)

@hussam-i-amhussam-i-am changed the title Auto complete css modulesfeat(Autocomplete): Convert Box usage to CSS modules behind feature flag Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5451 December 18, 2024 16:53 Inactive
@hussam-i-am
hussam-i-am marked this pull request as ready for review December 18, 2024 16:54
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerDecember 18, 2024 16:54
@hussam-i-am
hussam-i-am added this pull request to the merge queueDec 19, 2024
Merged via the queue into main with commit 59a1346Dec 19, 2024
@hussam-i-am
hussam-i-am deleted the auto-complete-css-modules branch December 19, 2024 16:25
@primerprimerBot mentioned this pull request Dec 19, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@randall-krauskopf@hussam-i-am@francinelucca