Skip to content

chore(deps): update @primer/octicons-react to 19.13.0 - #5456

Merged
langermank merged 8 commits into
mainfrom
bump-octicons
Dec 19, 2024
Merged

chore(deps): update @primer/octicons-react to 19.13.0#5456
langermank merged 8 commits into
mainfrom
bump-octicons

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Closes #

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Dec 18, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 89c5b2d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Dec 18, 2024
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Dec 18, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.27 KB (+0.22% 🔺)
packages/react/dist/browser.umd.js104.59 KB (+0.07% 🔺)

@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5456 December 18, 2024 20:18 Inactive
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5456 December 18, 2024 20:28 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 18, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5456 December 18, 2024 23:36 Inactive
@langermanklangermank added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Dec 19, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 19, 2024
@langermank
langermank marked this pull request as ready for review December 19, 2024 00:13
@langermank
langermank requested review from a team as code ownersDecember 19, 2024 00:13
@langermank
langermank added this pull request to the merge queueDec 19, 2024
Merged via the queue into main with commit e990fdbDec 19, 2024
@langermank
langermank deleted the bump-octicons branch December 19, 2024 00:40
@primerprimerBot mentioned this pull request Dec 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan