Skip to content

Pagination: Accessibility enhancements - #5488

Merged
hectahertz merged 4 commits into
mainfrom
hectahertz/pagination-a11y-enhancements
Jan 7, 2025
Merged

Pagination: Accessibility enhancements#5488
hectahertz merged 4 commits into
mainfrom
hectahertz/pagination-a11y-enhancements

Conversation

@hectahertz

@hectahertzhectahertz commented Jan 3, 2025

Copy link
Copy Markdown
Contributor

Changelog

New

Makes the default <Pagination/> story interactive:

interactive.mov

Changed

Hides the <Pagination/>Previous and Next buttons from the accessibility tree when they are disabled, as they are non functional.
Accessibility tree on Chrome devtools showing aria-hidden
Accessibility tree on Chrome devtools showing aria-hidden

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Things to review:

  • Logic of the status update on the Default example.
  • SR behavior concerning the Previous and Next buttons

Merge checklist

@hectahertz
hectahertz requested a review from a team as a code ownerJanuary 3, 2025 16:15
@changeset-bot

changeset-botBot commented Jan 3, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4dc4fbe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

This comment was marked as resolved.

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 3, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Jan 3, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.17 KB (-0.09% 🔽)
packages/react/dist/browser.umd.js104.65 KB (+0.01% 🔺)

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM. FYI we have a VRT issue on main we're investigating, so these failures are probably not related to your PR

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5488 January 7, 2025 15:17 Inactive
@hectahertz
hectahertz added this pull request to the merge queueJan 7, 2025
Merged via the queue into main with commit e352e19Jan 7, 2025
@hectahertz
hectahertz deleted the hectahertz/pagination-a11y-enhancements branch January 7, 2025 17:03
@primerprimerBot mentioned this pull request Jan 7, 2025
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/357068

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jan 9, 2025
francinelucca added a commit that referenced this pull request Jan 9, 2025
francinelucca added a commit that referenced this pull request Jan 9, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: failingChanges in this PR cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@hectahertz@francinelucca