Uh oh!
There was an error while loading. Please reload this page.
Avoid TypeError in PageLayout checking document.body - #5552
Merged
Conversation
Contributor
👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks! |
camertronforce-pushed
the
page_layout_check_body_null
branch
from
January 15, 2025 23:00
bdfba9c to
c01d5c8Compare🦋 Changeset detectedLatest commit: 5b8f1e0 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Contributor
size-limit report 📦
|
PageLayout checking document.bodycamertron
requested review from
francinelucca
and removed request for
mperrottiJanuary 16, 2025 18:58
francinelucca
approved these changes
Jan 16, 2025
👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/358254 |
🔴 golden-jobs completed with status |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for freeto join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Closeshttps://github.com/github/primer/issues/4634
Changelog
New
Changed
document.bodycan benull.Removed
Rollout strategy
Testing & Reviewing
Merge checklist
- [ ] Added/updated tests- [ ] Added/updated documentation- [ ] Added/updated previews (Storybook)- [ ] Tested in Chrome- [ ] Tested in Firefox- [ ] Tested in Safari- [ ] Tested in Edge