Skip to content

Banner allows critical variant to use ondismiss - #5583

Merged
joshblack merged 6 commits into
primer:mainfrom
brphelps:banner-allows-critical-ondismiss
Jan 23, 2025
Merged

Banner allows critical variant to use ondismiss#5583
joshblack merged 6 commits into
primer:mainfrom
brphelps:banner-allows-critical-ondismiss

Conversation

@brphelps

Copy link
Copy Markdown
Contributor

Closes#5582

Changelog

New

Changed

Banner now supports onDismiss used when using variant critical.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

I added a unit test and updated the example to use onDismiss.

Merge checklist

@brphelps
brphelps requested a review from a team as a code ownerJanuary 21, 2025 21:37
@changeset-bot

changeset-botBot commented Jan 21, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2206468

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks so much for putting this together! Just left a couple of comments 👍

Comment thread.changeset/blue-actors-rule.md Outdated
Comment threadpackages/react/src/Banner/Banner.features.stories.tsx Outdated
Co-authored-by: Josh Black <joshblack@github.com>
@joshblack
joshblack added this pull request to the merge queueJan 23, 2025
Merged via the queue into primer:main with commit db6c360Jan 23, 2025
@primerprimerBot mentioned this pull request Jan 23, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Allow critical Banner variant to use onDismiss

3 participants

@brphelps@lukasoppermann@joshblack