Uh oh!
There was an error while loading. Please reload this page.
refactor(ButtonBase): update selector to have specificty 0,1,0 - #5584
Conversation
🦋 Changeset detectedLatest commit: c78eae4 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks! |
size-limit report 📦
|
Uh oh!
There was an error while loading. Please reload this page.
👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/358837 |
🔴 golden-jobs completed with status |
mperrotti
left a comment
There was a problem hiding this comment.
Used the preview release to test in primer-docs, and this works as expected 🎉
Closeshttps://github.com/github/primer/issues/4703
Changelog
New
Changed
.ButtonBaseto have specificity of at least0,1,0to no longer get overridden by reset styles or equivalent selectors with0,1,0specificityRemoved
Rollout strategy