Skip to content

Move multiple component's feature flags from primer_react_css_modules_staff to primer_react_css_modules_ga - #5591

Merged
jonrohan merged 1 commit into
mainfrom
primer_css_modules_move_to_ga
Jan 23, 2025
Merged

Move multiple component's feature flags from primer_react_css_modules_staff to primer_react_css_modules_ga#5591
jonrohan merged 1 commit into
mainfrom
primer_css_modules_move_to_ga

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/4641

This moves multiple components from staff to ga

Changelog

New

Changed

  • Move Popover CSS modules feature flag from staff to ga
  • Move Textarea CSS modules feature flag from staff to ga
  • Move UnstyledTextInput CSS modules feature flag from staff to ga
  • Move SelectPanel2 CSS modules feature flag from staff to ga
  • Move Breadcrumbs CSS modules feature flag from staff to ga
  • Move ProgressBar CSS modules feature flag from staff to ga
  • Move Dialog CSS modules feature flag from staff to ga
  • Move DialogV1 CSS modules feature flag from staff to ga
  • Move TreeView CSS modules feature flag from staff to ga
  • Move CheckboxOrRadioGroup CSS modules feature flag from staff to ga

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerJanuary 23, 2025 18:03
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ff40a93

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jan 23, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.54 KB (+0.01% 🔺)
packages/react/dist/browser.umd.js104.86 KB (-0.05% 🔽)

@jonrohan
jonrohan added this pull request to the merge queueJan 23, 2025
Merged via the queue into main with commit 12f1af7Jan 23, 2025
@jonrohan
jonrohan deleted the primer_css_modules_move_to_ga branch January 23, 2025 19:14
@primerprimerBot mentioned this pull request Jan 23, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@TylerJDev