Skip to content

fix(useResizeObserver): undefined variable alternative - #5595

Merged
francinelucca merged 5 commits into
mainfrom
fix/resize-observer-patch
Jan 29, 2025
Merged

fix(useResizeObserver): undefined variable alternative#5595
francinelucca merged 5 commits into
mainfrom
fix/resize-observer-patch

Conversation

@francinelucca

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4685

We were seeing a small number of errors for undefined ResizeObserver, introducing a check for when ResizeObserver is not defined and using a manual solution instead.

Changelog

Changed

  • Detect when ResideObserver is not defined and use manual solution instead

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@francinelucca
francinelucca requested a review from a team as a code ownerJanuary 25, 2025 01:34
@changeset-bot

changeset-botBot commented Jan 25, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ecbf35c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Jan 25, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 25, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5595 January 25, 2025 01:36 Inactive
@github-actions

github-actionsBot commented Jan 25, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.51 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js104.9 KB (+0.05% 🔺)

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Jan 25, 2025

@camertroncamertron left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM 🎉 Looks like there might be a few cases where we don't need to check for null/undefined tho?

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/359330

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Jan 29, 2025
@francinelucca
francinelucca added this pull request to the merge queueJan 29, 2025
Merged via the queue into main with commit 2db6c92Jan 29, 2025
@francinelucca
francinelucca deleted the fix/resize-observer-patch branch January 29, 2025 05:43
@primerprimerBot mentioned this pull request Jan 29, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@camertron